Skip to content

Remove Beta Label / Warning from Encryption At Rest - #7602

Merged
kubermatic-bot merged 2 commits into
mainfrom
hotfix/remove-encryption-warning-and-beta-label
Oct 1, 2025
Merged

kubermatic-bot merged 2 commits into
mainfrom
hotfix/remove-encryption-warning-and-beta-label

Conversation

@KhizerRehan

Copy link
Copy Markdown
Contributor

What this PR does / why we need it:
This PR removes warning/beta label from Encryption at Rest PR.

Do you want me to make it more formal (for review notes) or short and to the point (for commit/PR description)?

Before: With Warning/Beta Label
Screenshot 2025-09-22 at 11 14 45 AM

After:
image

Which issue(s) this PR fixes:

Fixes #

What type of PR is this?

Special notes for your reviewer:

Does this PR introduce a user-facing change? Then add your Release Note here:

NONE

Documentation:

NONE

@kubermatic-bot kubermatic-bot added release-note-none Denotes a PR that doesn't merit a release note. do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. docs/none Denotes a PR that doesn't need documentation (changes). dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. labels Sep 22, 2025
@kubermatic-bot

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@kubermatic-bot kubermatic-bot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Sep 22, 2025
Comment thread modules/web/src/assets/css/global/_main.scss Outdated
@KhizerRehan
KhizerRehan marked this pull request as ready for review September 29, 2025 20:10
@kubermatic-bot kubermatic-bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 29, 2025
@codecov

codecov Bot commented Sep 29, 2025 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 45.01%. Comparing base (74f8bb3) to head (8d5a34f).
⚠️ Report is 9 commits behind head on main.

Additional details and impacted files
@@            Coverage Diff             @@
##             main    #7602      +/-   ##
==========================================
- Coverage   45.03%   45.01%   -0.02%     
==========================================
  Files         441      441              
  Lines       19814    19814              
  Branches     2751     3214     +463     
==========================================
- Hits         8923     8920       -3     
+ Misses      10888    10264     -624     
- Partials        3      630     +627     

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@Waseem826 Waseem826 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@kubermatic-bot kubermatic-bot added the lgtm Indicates that a PR is ready to be merged. label Oct 1, 2025
@kubermatic-bot

Copy link
Copy Markdown
Contributor

LGTM label has been added.

DetailsGit tree hash: 1d53488033a8087eedd7402a7205bbd8f852614e

@kubermatic-bot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Waseem826

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubermatic-bot kubermatic-bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 1, 2025
@Waseem826

Copy link
Copy Markdown
Contributor

@KhizerRehan can you also please remove the beta warning message from docs here: https://docs.kubermatic.com/kubermatic/main/tutorials-howtos/encryption-at-rest/

@kubermatic-bot
kubermatic-bot merged commit 5cf0f34 into main Oct 1, 2025
9 of 10 checks passed
@kubermatic-bot kubermatic-bot added this to the KKP 2.29 milestone Oct 1, 2025
@kubermatic-bot
kubermatic-bot deleted the hotfix/remove-encryption-warning-and-beta-label branch October 1, 2025 09:00
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. dco-signoff: yes Denotes that all commits in the pull request have the valid DCO signoff message. docs/none Denotes a PR that doesn't need documentation (changes). lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note. size/M Denotes a PR that changes 30-99 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants