Conversation
|
/assign @johnbelamaric |
| @@ -31,8 +31,9 @@ latest-milestone: "v1.35" | |||
| # The milestone at which this feature was, or is targeted to be, at each stage. | |||
| milestone: | |||
| alpha: "v1.34" | |||
| beta: "v1.35" | |||
| stable: "v1.36" | |||
| alpha: "v1.35" | |||
There was a problem hiding this comment.
I think you can leave 1.34 here - that's when it first entered alpha.
There was a problem hiding this comment.
1.34 is still there, at line 33
There was a problem hiding this comment.
oh, we can't have it twice, it's a map key
There was a problem hiding this comment.
i see. removed the alpha 1.35
|
/hold for scheduling approvers |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: johnbelamaric, yliaog The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/close |
|
@macsko: Closed this PR. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
changed 1.36 to beta for DRA extended resource feature
DRA: Handle extended resource requests via DRA Driver #5004