Repository navigation
docs: tweak minimum-reminder-interval description - #1194
Merged
Merged
Conversation
TwiN
approved these changes
Aug 11, 2025
Owner
|
Thank you for the contribution |
alexlebens
pushed a commit
to alexlebens/infrastructure
that referenced
this pull request
Aug 18, 2025
This PR contains the following updates: | Package | Update | Change | |---|---|---| | [ghcr.io/twin/gatus](https://github.com/TwiN/gatus) | minor | `v5.22.0` -> `v5.23.1` | --- ### Release Notes <details> <summary>TwiN/gatus (ghcr.io/twin/gatus)</summary> ### [`v5.23.1`](https://github.com/TwiN/gatus/releases/tag/v5.23.1) [Compare Source](TwiN/gatus@v5.23.0...v5.23.1) #### What's Changed - fix(ui): Update endpoints in-place instead of reloading every time by [@​TwiN](https://github.com/TwiN) in TwiN/gatus#1209 - docs: improve `extra-labels` description by [@​salim-b](https://github.com/salim-b) in TwiN/gatus#1208 **Full Changelog**: TwiN/gatus@v5.23.0...v5.23.1 ### [`v5.23.0`](https://github.com/TwiN/gatus/releases/tag/v5.23.0) [Compare Source](TwiN/gatus@v5.22.0...v5.23.0) This is a big one, and for the occasion, I'll highlight the two big changes: - **New UI**: That's right, a brand new modern UI. This was long overdue. I hope you like it. - **Announcements**: Want to update your users/customers about an outage? You can, now.  Thoughts on the UI & announcements? Just like any healthy relationship, communicate with me on [TwiN/gatus#1206](TwiN/gatus#1206) #### What's Changed - feat(ui): New status page UI by [@​TwiN](https://github.com/TwiN) in TwiN/gatus#1198 - feat: Implement announcements by [@​TwiN](https://github.com/TwiN) in TwiN/gatus#1204 - feat(alerting): Support extra labels in email alerts by [@​appleboy](https://github.com/appleboy) in TwiN/gatus#1191 - fix(ui): Replace filter value "nothing" by "none" by [@​TwiN](https://github.com/TwiN) in TwiN/gatus#1202 - docs: tweak `minimum-reminder-interval` description by [@​salim-b](https://github.com/salim-b) in TwiN/gatus#1194 **Full Changelog**: TwiN/gatus@v5.22.0...v5.23.0 </details> --- ### Configuration 📅 **Schedule**: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined). 🚦 **Automerge**: Disabled by config. Please merge this manually once you are satisfied. ♻ **Rebasing**: Whenever PR is behind base branch, or you tick the rebase/retry checkbox. 🔕 **Ignore**: Close this PR and you won't be reminded about this update again. --- - [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box --- This PR has been generated by [Renovate Bot](https://github.com/renovatebot/renovate). <!--renovate-debug:eyJjcmVhdGVkSW5WZXIiOiI0MS4zNS4wIiwidXBkYXRlZEluVmVyIjoiNDEuMS4zIiwidGFyZ2V0QnJhbmNoIjoibWFpbiIsImxhYmVscyI6WyJpbWFnZSJdfQ==--> Reviewed-on: https://gitea.alexlebens.dev/alexlebens/infrastructure/pulls/1241 Co-authored-by: Renovate Bot <renovate-bot@alexlebens.net> Co-committed-by: Renovate Bot <renovate-bot@alexlebens.net>
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Summary
Improve documentation of
alerts[].minimum-reminder-interval, cf. #1138.Checklist
README.md, if applicable.