Repository navigation
feat(playwright,cypress): bump version to v1 - #454
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: ebe4f61 The changes in this PR will be included in the next version bump. This PR includes changesets to release 2 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
commit: |
JReinhold
approved these changes
Sep 24, 2026
Member
|
@AriPerkkio can you share some context on why now is the time to go to v1 for these packages? (Not that I oppose, just making sure I have the context of where features are at). |
Member
Author
|
The public API of these packages has been stable for long time, so they've been ready for stable release for long. And now that #445 has landed I can't think of any other breaking changes that would block us from v1. So there really isn't any reason why not release v1. 😄 |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What Changed
Prepare
@chromatic-com/playwrightand@chromatic-com/cypressfor their first major version releases.How to test